Update module sigs.k8s.io/controller-runtime to v0.25.1 - #2072
renovate-bot wants to merge 1 commit into
Conversation
|
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: renovate-bot The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
|
@renovate-bot: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
This pull request has been automatically marked as stale because it has not had recent activity. It will be closed in 30 days if no further activity occurs. Thank you for your contributions. |
3e892df to
6aa1ab3
Compare
Signed-off-by: Mend Renovate <bot@renovateapp.com>
6aa1ab3 to
1188949
Compare
|



This PR contains the following updates:
v0.23.3→v0.25.1Release Notes
kubernetes-sigs/controller-runtime (sigs.k8s.io/controller-runtime)
v0.25.1Compare Source
What's Changed
Full Changelog: kubernetes-sigs/controller-runtime@v0.25.0...v0.25.1
v0.25.0Compare Source
Highlights
This version of controller-runtime introduces a new experimental
ReadYourWritesConsistencyfeature,which ensures that all writes are reflected in subsequent reads from the default cache-backed client.
Stale client reads are arguably the biggest source of friction and sometimes bugs for controller
authors, providing this functionality at the library level eliminates that class of problems entirely.
Try it out by setting
Client.EnableReadYourWritesConsistency: new(true)in your managerand leave any feedback you may have on the tracking issue.
✨ New Features
🐛 Bug Fixes
🌱 Others
🌱 CI & linters
📖 Additionally, there have been 7 contributions to our documentation and book. (#3513, #3521, #3531, #3551, #3562, #3563, #3566)
Dependencies
Added
Changed
944ab1f→746e56fbd525da→59b49668636f87→3dc84a48636f87→3dc84a4ec3ebc5→25e220843fb72c→d427ff9b8788ab→be93311Removed
Thanks to all our contributors! 😊
v0.24.1Compare Source
What's Changed
Full Changelog: kubernetes-sigs/controller-runtime@v0.24.0...v0.24.1
v0.24.0Compare Source
🐛 Bug Fixes
🌱 Others
StartLoggingfor event debug logs (#3451)🌱 CI & linters
tools/setup-envtestsubmodule (#3476)📖 Additionally, there has been 1 contribution to our documentation. (#3477)
Dependencies
Added
Changed
8a7402a→944ab1f078029d→bd525daa0af3ef→8636f87200df99→8636f87f2248ac589584f→43fb72cbc988d5→b8788abRemoved
9bdfabeThanks to all our contributors! 😊
Configuration
📅 Schedule: (in timezone America/New_York)
* 3-6 * * 1-5)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.