Skip to content

Fix non-constant format string error on go 1.26 - #3392

Open
dsimansk wants to merge 1 commit into
knative:mainfrom
dsimansk:pr/fix-format-string
Open

dsimansk wants to merge 1 commit into
knative:mainfrom
dsimansk:pr/fix-format-string

Conversation

@dsimansk

Copy link
Copy Markdown
Contributor

Changes

  • Fix non-constant format string error on go 1.26

Bumping go directive in go.mod to force error fix verification through unit tests. The 1.26's go vet is has tighter recognition of string formatters not being present.

testing/duck/testbindable.go:153:63: non-constant format string in call to (knative.dev/pkg/apis.ConditionManager).MarkFalse

Appeared here: #3391

Release Note

Fix non-constant format string error on go 1.26

/cc @gauron99 @kahirokunn

@knative-prow knative-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 17, 2026
@knative-prow

knative-prow Bot commented Sep 17, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dsimansk

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 17, 2026
@dsimansk

Copy link
Copy Markdown
Contributor Author

There's couple more occasions of same use across our repos: #3387

@codecov

codecov Bot commented Sep 17, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 74.88%. Comparing base (d2a153a) to head (3ad1cca).

Additional details and impacted files
@@            Coverage Diff             @@
##             main    #3392      +/-   ##
==========================================
+ Coverage   74.85%   74.88%   +0.02%     
==========================================
  Files         189      189              
  Lines        8349     8349              
==========================================
+ Hits         6250     6252       +2     
+ Misses       1854     1853       -1     
+ Partials      245      244       -1     

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant