Skip to content

Add CODEOWNERS for automatic review assignment - #16449

Merged
diox merged 1 commit into
mainfrom
codeowners
Sep 11, 2026
Merged

Add CODEOWNERS for automatic review assignment#16449
diox merged 1 commit into
mainfrom
codeowners

Conversation

@diox

@diox diox commented Sep 11, 2026

Copy link
Copy Markdown
Member

No description provided.

@diox
diox requested a review from willdurand September 11, 2026 09:14
@diox
diox merged commit 00817a7 into main Sep 11, 2026
43 of 62 checks passed
@diox
diox deleted the codeowners branch September 11, 2026 09:38
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants