[release-v1.23] Update Konflux configurations - #944
Open
serverless-qe wants to merge 2 commits into
Open
serverless-qe wants to merge 2 commits into
serverless-qe wants to merge 2 commits into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: serverless-qe The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
serverless-qe
force-pushed
the
sync-konflux-release-v1.23
branch
3 times, most recently
from
August 31, 2026 13:14
494433e to
1027deb
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.23
branch
15 times, most recently
from
September 7, 2026 11:57
f1ab5b8 to
fc25f06
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.23
branch
4 times, most recently
from
September 14, 2026 12:17
ed6535c to
6020889
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.23
branch
5 times, most recently
from
September 16, 2026 15:39
cac4c53 to
285445b
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.23
branch
5 times, most recently
from
September 18, 2026 04:37
ef80cb6 to
4ecc21e
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.23
branch
from
September 18, 2026 10:36
4ecc21e to
8deeb54
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Update Konflux configurations