Conversation
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: maschmid The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/hold testing |
|
DNM in openshift/release#79402 , looks fine |
|
/unhold |
| module github.com/openshift-knative/hack | ||
|
|
||
| go 1.25.5 | ||
| go 1.25.8 |
There was a problem hiding this comment.
➜ ~ podman run --pull always -it registry.ci.openshift.org/openshift/release:rhel-9-release-golang-1.25-openshift-4.21 go version
Trying to pull registry.ci.openshift.org/openshift/release:rhel-9-release-golang-1.25-openshift-4.21...
Getting image source signatures
Copying blob sha256:9fa64130ebc4553b0ccb4de57407e9cad16c5aaa9d68c2903f57afe60508fd11
Copying blob sha256:65442b90c1456946b65a513148c3e504c3e1ae3889d427d5b4526ba9cec3ea0e
Copying blob sha256:773e4463085455d5c3b616c073a7844366ef1ff55cb12a9c5d694891d45644eb
Copying blob sha256:521de6e2ed83d650041cfd23f3495f5fc6bcfdb2cd936b0598fa4d2c90982d9f
Copying config sha256:7f126589239d41fe5849640ef53a07b7a8a1b793c6f62a7be1582b597cbc155d
Writing manifest to image destination
go version go1.25.7 (Red Hat 1.25.7-1.el9_7) linux/arm64
/hold
There was a problem hiding this comment.
➜ ~ podman run --pull always -it registry.ci.openshift.org/openshift/release:rhel-9-release-golang-1.25-openshift-4.22 go version
...
go version go1.25.9 (Red Hat 1.25.9-1.el9_7) linux/arm64
There was a problem hiding this comment.
hm, Claude says "Both sigs.k8s.io/prow and github.com/openshift/ci-tools declare go 1.25.8 in their go.mod."
I guess we'll hold until we can actually use it then :)
|
PR needs rebase. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
..the downside is we now need to manually add the
reporter_configto the manually defined custom tests, but that's overall cleaner anyway.