Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Build failed (check pipeline). Post ✔️ s2i-openstack-containers-molecule SUCCESS in 2m 06s |
|
Tested in https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/f3c2bc42c4e541a98b42dec056865319 . Tempest test failure is known issue and unrelated but the API is working fine. |
|
check-rdo |
|
Build succeeded (check pipeline). ✔️ s2i-openstack-containers-molecule SUCCESS in 2m 05s |
| # backwards compatibility | ||
|
|
||
| RUN rm -f /usr/bin/watcher-api-wsgi && \ | ||
| ln -s /usr/lib/python3.12/site-packages/watcher/wsgi/api.py /usr/bin/watcher-api-wsgi |
There was a problem hiding this comment.
Do we need to parameterise python version for future?
There was a problem hiding this comment.
if we avoid hardcoding it will also be useful for the future if we use the same file for multiple base images with different python versions
There was a problem hiding this comment.
These containerfiles are intended to build only for EL10 with python312 so i didn't want to complicate it much tbh. Said so, we may implement an easy way to manage the python version with:
RUN rm -f /usr/bin/watcher-api-wsgi && \
ln -s /usr/lib/python*/site-packages/watcher/wsgi/api.py /usr/bin/watcher-api-wsgi && \
[ -e "/usr/bin/watcher-api-wsgi" ]
What do you think?
There was a problem hiding this comment.
I was checking in nova-api and they added scripts https://github.com/openstack-k8s-operators/s2i-openstack-containers/tree/main/containers/nova/nova-api/scripts which are copied to /usr/bin:
s2i-openstack-containers/containers/nova/nova-api/Containerfile
Lines 102 to 103 in 760a77e
There was a problem hiding this comment.
We want to avoid maintaining a python wrapper script in the repo needlessly, that's why i proposed to do the symlink which should be easier to maintain
viroel
left a comment
There was a problem hiding this comment.
from zuul job:
[watcher/watcher-base] [2/2] STEP 12/15: RUN rm -f /usr/bin/watcher-api-wsgi && ln -s /usr/lib/python3.12/site-packages/watcher/wsgi/api.py /usr/bin/watcher-api-wsgi
In the end the validation job doesn't deploy watcher, so I was able to check that it build only, but we don't have a validation. |
I tested before with a depends-on on #203 which is not merged yet. That's the log i pasted in a previous comment. Logs are in https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/ecbcbfaa5eac4a03836d72216d59490d Failure in tempest is unrelated and has been fixed separatedly. It could be executed again if desired. |
ok, sorry, i missed your comment there. |
As part of eventlet removal, watcher is removing the deprecated wsgi script /usr/bin/watcher-api-wsgi which is the path used as WSGIScriptAlias [1]. This patch is creating a symlink in that path to the watcher/wsgi/api.py which is valid to mod_wsgi to maintain a backwards compatible interface. [1] https://review.opendev.org/c/openstack/watcher/+/1006113/ Signed-off-by: Alfredo Moralejo <amoralej@redhat.com>
|
This change depends on a change that failed to merge. Change #203 is needed. |
|
check-rdo |
|
Build succeeded (check pipeline). ✔️ watcher-s2i-tempest SUCCESS in 1h 57m 44s |
As part of eventlet removal, watcher is removing the deprecated wsgi script /usr/bin/watcher-api-wsgi which is the path used as WSGIScriptAlias [1].
This patch is creating a symlink in that path to the watcher/wsgi/api.py which is valid to mod_wsgi to maintain a backwards compatible interface.
[1] https://review.opendev.org/c/openstack/watcher/+/1006113/
(adding depends-on to trigger watcher deployment job which is not merged yet)
Depends-On: #203