Skip to content

Update HorizonTest image name variable - #527

Merged
openshift-merge-bot[bot] merged 1 commit into
openstack-k8s-operators:mainfrom
kstrenkova:rename-horizon-test-image-var
Sep 25, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
openstack-k8s-operators:mainfrom
kstrenkova:rename-horizon-test-image-var

Conversation

@kstrenkova

Copy link
Copy Markdown
Contributor

The variable RELATE_IMAGE_HORIZONTEST was not consistent with other resource types that also have TEST_ infix. This change makes it consistent not only for HorizonTest and fixed a small bug, but also with other resource types.

The variable RELATE_IMAGE_HORIZONTEST was not consistent with
other resource types that also have TEST_ infix. This change
makes it consistent not only for HorizonTest and fixed a small
bug, but also with other resource types.
@coderabbitai

coderabbitai Bot commented Sep 25, 2026

Copy link
Copy Markdown

Important

  • 🔍 Trigger review

This repository does not receive automatic reviews because it has fewer than 10 stars.

⚙️ Run configuration

Configuration used: Central YAML (base), Organization UI (inherited)

Review profile: CHILL

Plan: Advanced

Run ID: d54cacd2-4aef-436e-aa8d-71adbd2c3ca9


Comment @coderabbitai help to get the list of available commands.

@kstrenkova kstrenkova added the ready-for-review Marks additional review needed label Sep 25, 2026
@openshift-ci

openshift-ci Bot commented Sep 25, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: adrianfusco, kstrenkova

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [adrianfusco,kstrenkova]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build succeeded (check pipeline).
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/f3d175104cbf489e9ab94c26d90ab5c4

✔️ test-operator-s2i-content-provider SUCCESS in 3h 25m 02s (non-voting)
✔️ test-operator-kuttl-s2i SUCCESS in 37m 16s (non-voting)
✔️ test-operator-s2i-test SUCCESS in 2h 30m 17s (non-voting)
✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 22m 54s
✔️ test-operator-kuttl SUCCESS in 39m 56s
✔️ podified-multinode-edpm-deployment-crc-test-operator SUCCESS in 2h 45m 59s

@openshift-merge-bot
openshift-merge-bot Bot merged commit 8d2cc37 into openstack-k8s-operators:main Sep 25, 2026
9 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved lgtm ready-for-review Marks additional review needed

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants