Skip to content

[docs] Update images and add s2i variants - #528

Draft
kstrenkova wants to merge 1 commit into
openstack-k8s-operators:mainfrom
kstrenkova:update-docs-with-s2i-images
Draft

kstrenkova wants to merge 1 commit into
openstack-k8s-operators:mainfrom
kstrenkova:update-docs-with-s2i-images

Conversation

@kstrenkova

Copy link
Copy Markdown
Contributor

This change is a documentation update following the new s2i images support. The tcib images can still be used, so the wording is update a bit and explanation for the new s2i images is added.

This change is a documentation update following the new s2i
images support. The tcib images can still be used, so the wording
is update a bit and explanation for the new s2i images is added.
@openshift-ci

openshift-ci Bot commented Sep 25, 2026

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@openshift-ci

openshift-ci Bot commented Sep 25, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: kstrenkova

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@coderabbitai

coderabbitai Bot commented Sep 25, 2026

Copy link
Copy Markdown

Important

Draft PR not reviewed

Draft PRs are not automatically reviewed by default.

  • Trigger a manual review

To automatically review draft PRs, update your CodeRabbit configuration:

reviews:
  auto_review:
    drafts: true

Comment @coderabbitai help to get the list of available commands.

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build succeeded (check pipeline).
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/70b414a74b6e43dda1422607e1997697

✔️ test-operator-s2i-content-provider SUCCESS in 3h 12m 58s (non-voting)
✔️ test-operator-kuttl-s2i SUCCESS in 36m 04s (non-voting)
✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 49m 41s
✔️ test-operator-kuttl SUCCESS in 38m 20s

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant