Conversation
Depends-On: openstack-k8s-operators/openstack-operator#2070 Signed-off-by: Alfredo Moralejo <amoralej@redhat.com>
|
Important
This repository does not receive automatic reviews because it has fewer than 10 stars. ⚙️ Run configurationConfiguration used: Central YAML (base), Organization UI (inherited) Review profile: CHILL Plan: Advanced Run ID: Comment |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 17m 32s |
|
legit issue found when deploying with s2i containers. I guess it's missing dependency in openstack-nova-compute |
|
check-rdo |
|
Build succeeded (check pipeline). ✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 25m 14s |
Depends-On: openstack-k8s-operators/openstack-operator#2070