chore(deps): bump the all group in /release with 3 updates - #1973
Merged
Merged
Conversation
Bumps the all group in /release with 3 updates: [chainguard/crane](https://github.com/chainguard-images/images), [tektoncd/plumbing/ko](https://github.com/tektoncd/plumbing) and [tektoncd/plumbing/koparse](https://github.com/tektoncd/plumbing). Updates `chainguard/crane` from `159ce4f` to `5db7b26` - [Commits](https://github.com/chainguard-images/images/commits) Updates `tektoncd/plumbing/ko` from `579038f` to `90b7fd2` - [Commits](https://github.com/tektoncd/plumbing/commits) Updates `tektoncd/plumbing/koparse` from `205c228` to `73a2f2d` - [Commits](https://github.com/tektoncd/plumbing/commits) --- updated-dependencies: - dependency-name: chainguard/crane dependency-version: latest-dev dependency-type: direct:production dependency-group: all - dependency-name: tektoncd/plumbing/ko dependency-version: 90b7fd2b7c7e5358e7ce1ffa9a5e557dd2dfef27711a6950c302e10368f1518f dependency-type: direct:production dependency-group: all - dependency-name: tektoncd/plumbing/koparse dependency-version: 73a2f2da52c7314616322fcd0a1f10d0ae44a8aeac08a2b36deea147aed0e208 dependency-type: direct:production dependency-group: all ... Signed-off-by: dependabot[bot] <support@github.com>
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #1973 +/- ##
==========================================
- Coverage 72.60% 72.55% -0.06%
==========================================
Files 58 58
Lines 3855 3855
==========================================
- Hits 2799 2797 -2
- Misses 763 764 +1
- Partials 293 294 +1
Flags with carried forward coverage won't be shown. Click here to find out more. ☔ View full report in Codecov by Harness. 🚀 New features to boost your workflow:
|
Contributor
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: infernus01 The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Warning
Cooldown could not be applied because no publication date was available from the registry.
Bumps the all group in /release with 3 updates: chainguard/crane, tektoncd/plumbing/ko and tektoncd/plumbing/koparse.
Updates
chainguard/cranefrom159ce4fto5db7b26Commits
Updates
tektoncd/plumbing/kofrom579038fto90b7fd2Commits
Updates
tektoncd/plumbing/koparsefrom205c228to73a2f2dCommits
Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore <dependency name> major versionwill close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)@dependabot ignore <dependency name> minor versionwill close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)@dependabot ignore <dependency name>will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)@dependabot unignore <dependency name>will remove all of the ignore conditions of the specified dependency@dependabot unignore <dependency name> <ignore condition>will remove the ignore condition of the specified dependency and ignore conditions