Skip to content

chore(deps): bump the all group in /release with 3 updates - #1981

Open
dependabot[bot] wants to merge 1 commit into
release-v0.28.xfrom
dependabot/docker/release/release-v0.28.x/all-759229a093
Open

dependabot[bot] wants to merge 1 commit into
release-v0.28.xfrom
dependabot/docker/release/release-v0.28.x/all-759229a093

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 16, 2026

Copy link
Copy Markdown
Contributor

Warning

Cooldown could not be applied because no publication date was available from the registry.

Bumps the all group in /release with 3 updates: chainguard/crane, tektoncd/plumbing/ko and tektoncd/plumbing/koparse.

Updates chainguard/crane from 159ce4f to 5db7b26

Commits

Updates tektoncd/plumbing/ko from 579038f to 90b7fd2

Commits

Updates tektoncd/plumbing/koparse from 205c228 to 73a2f2d

Commits

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore <dependency name> major version will close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)
  • @dependabot ignore <dependency name> minor version will close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)
  • @dependabot ignore <dependency name> will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)
  • @dependabot unignore <dependency name> will remove all of the ignore conditions of the specified dependency
  • @dependabot unignore <dependency name> <ignore condition> will remove the ignore condition of the specified dependency and ignore conditions

@dependabot dependabot Bot added dependencies Used by dependabot - identifies all PRs created by dependabot kind/misc Categorizes issue or PR as a miscellaneuous one. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note-none Denotes a PR that doesnt merit a release note. labels Sep 16, 2026
@tekton-robot tekton-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 16, 2026
@tekton-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: anithapriyanatarajan

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [anithapriyanatarajan]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@tekton-robot tekton-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 17, 2026
Bumps the all group in /release with 3 updates: [chainguard/crane](https://github.com/chainguard-images/images), [tektoncd/plumbing/ko](https://github.com/tektoncd/plumbing) and [tektoncd/plumbing/koparse](https://github.com/tektoncd/plumbing).


Updates `chainguard/crane` from `159ce4f` to `5db7b26`
- [Commits](https://github.com/chainguard-images/images/commits)

Updates `tektoncd/plumbing/ko` from `579038f` to `90b7fd2`
- [Commits](https://github.com/tektoncd/plumbing/commits)

Updates `tektoncd/plumbing/koparse` from `205c228` to `73a2f2d`
- [Commits](https://github.com/tektoncd/plumbing/commits)

---
updated-dependencies:
- dependency-name: chainguard/crane
  dependency-version: latest-dev
  dependency-type: direct:production
  dependency-group: all
- dependency-name: tektoncd/plumbing/ko
  dependency-version: 90b7fd2b7c7e5358e7ce1ffa9a5e557dd2dfef27711a6950c302e10368f1518f
  dependency-type: direct:production
  dependency-group: all
- dependency-name: tektoncd/plumbing/koparse
  dependency-version: 73a2f2da52c7314616322fcd0a1f10d0ae44a8aeac08a2b36deea147aed0e208
  dependency-type: direct:production
  dependency-group: all
...

Signed-off-by: dependabot[bot] <support@github.com>
@anithapriyanatarajan
anithapriyanatarajan force-pushed the dependabot/docker/release/release-v0.28.x/all-759229a093 branch from b80ab21 to 6831467 Compare September 17, 2026 01:59
@anithapriyanatarajan

Copy link
Copy Markdown
Contributor

/lgtm

@tekton-robot tekton-robot added the lgtm Indicates that a PR is ready to be merged. label Sep 17, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. dependencies Used by dependabot - identifies all PRs created by dependabot kind/misc Categorizes issue or PR as a miscellaneuous one. lgtm Indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note-none Denotes a PR that doesnt merit a release note. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants