Skip to content

Add a Watcher job to the S2I tests - #203

Open
rlandy wants to merge 1 commit into
openstack-k8s-operators:mainfrom
rlandy:watcher-test
Open

rlandy wants to merge 1 commit into
openstack-k8s-operators:mainfrom
rlandy:watcher-test

Conversation

@rlandy

@rlandy rlandy commented Sep 15, 2026 •

Copy link
Copy Markdown

This PR adds a full deployment job with Watcher.

@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign arxcruz for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@centosinfra-prod-github-app

Copy link
Copy Markdown

Zuul encountered a syntax error while parsing its
configuration in the repo openstack-k8s-operators/s2i-openstack-containers on branch main. The
problem was:

while parsing a block collection
in "openstack-k8s-operators/s2i-openstack-containers/zuul.d/jobs.yaml@main", line 226, column 7
did not find expected '-' indicator
in "openstack-k8s-operators/s2i-openstack-containers/zuul.d/jobs.yaml@main", line 230, column 7

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/ef45585b056d42eb89d5379244cc559a

✔️ s2i-openstack-containers-molecule SUCCESS in 2m 10s
❌ s2i-openstack-container-content-provider NODE_FAILURE Node(set) request 100-0000203664 failed in 0s
⚠️ s2i-openstack-container-consumer-smoke SKIPPED Skipped due to failed job s2i-openstack-container-content-provider
⚠️ s2i-openstack-deploy-validation SKIPPED Skipped due to failed job s2i-openstack-container-content-provider
⚠️ s2i-openstack-deploy-validation-watcher SKIPPED Skipped due to failed job s2i-openstack-container-content-provider

@rlandy

rlandy commented Sep 15, 2026

Copy link
Copy Markdown
Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/31b5f45732cf4f5e99626ad3da610783

✔️ s2i-openstack-containers-molecule SUCCESS in 2m 04s
❌ s2i-openstack-container-content-provider FAILURE in 4m 54s
⚠️ s2i-openstack-container-consumer-smoke SKIPPED Skipped due to failed job s2i-openstack-container-content-provider
⚠️ s2i-openstack-deploy-validation SKIPPED Skipped due to failed job s2i-openstack-container-content-provider
⚠️ s2i-openstack-deploy-validation-watcher SKIPPED Skipped due to failed job s2i-openstack-container-content-provider

@amoralej

amoralej commented Sep 15, 2026 •

Copy link
Copy Markdown
Contributor

it seems centos10 update broke it, i'm checking what the issue is.

@amoralej

Copy link
Copy Markdown
Contributor

it seems centos10 update broke it, i'm checking what the issue is.

UBI10 has openssl-libs-3.5.8-1.el10_2.x86_64 which is newer from nvr that the centos one which is openssl-libs-1:3.5.8-1.el10.x86_64

Comment thread zuul.d/projects.yaml Outdated
- s2i-openstack-container-content-provider
- s2i-openstack-container-consumer-smoke
- s2i-openstack-deploy-validation-watcher:
irrelevant-files: *irrelevant-for-zuul

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

so ideealy instead of addin ght job directly we would add it to a watcher porject template and we need to have this trigger on its own file list

sepeicifly it shoudl triger on change to teh base contaienr or watcher contaienr folders but not all folders so

irrelevant-files: *irrelevant-for-zuul

is not the correct approch, we want to invert it and only list the subpath that we need the job to run on

Comment thread zuul.d/jobs.yaml Outdated
run: playbooks/container-ci/zuul/consumer-smoke.yaml

- job:
name: s2i-openstack-deploy-validation-watcher

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

we also shoudl not be defineing the job in this repo
the defintion should be in the watcher-operator repo and only reused form this repo

Copy link
Copy Markdown
Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@amoralej do you also agree that that this job should be in the watcher-operator repo? If so, will close this and redefine there

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

you would sill need this change to add the job defiention to trigger on this repo
the intent is to have a common job defieiotn in the watcher-oeprator repo htat can trigger on changes to both repos based on diffetn set of files

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I was thinking in adding it here to test to avoid having two changes with depends-on, etc... and make it easier. At the end, yep, it may be good to define the job or maybe a template in watcher-operator repo.

Copy link
Copy Markdown
Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

ok - will define the job in watcher-operator - and then change this PR to reference it

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

so we have watcher-s2i-tempest already
https://github.com/openstack-k8s-operators/watcher-operator/blob/main/.zuul.yaml#L139-L181

which currently inherits form watcher-operator-validation-base which exented watcher-operator-base and that currently inherites form podified-multinode-edpm-deployment-crc-2comp

https://github.com/openstack-k8s-operators/watcher-operator/blob/main/.zuul.yaml#L60-L87

https://github.com/openstack-k8s-operators/watcher-operator/blob/main/.zuul.yaml#L2-L20

my expecation is watcher-operator-base will be update to inherit form
s2i-test-base in the next week or two so that all out tempest jobs derive form that as its baseline, and here woudl just run watcher-s2i-tempest here with overriden files:

we do not need a duplicate job defiention becuase zuul supprot job variant so when you add watcher-s2i-tempest to github-check via a templeate or directly you can add it as

github-check:
   watcher-s2i-tempest:
      files:
        - containers/watcher
        - containers/base
      dependencies:
        - s2i-openstack-container-content-provider
        - s2i-openstack-container-consumer-smoke

and we are done from the point of view of this repo

all of the rest lives in -operator

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

by the way the only reason i was suggesting a template is to turnt that 7 line addtion to the project into a one line addtion of the template and alow you to have that templelate out of line but its its not strictly required

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

We discussed in last s2i call about the patter to add coverage for deployment specific scenarios to s2i. As mentioned here, we should be creating a template in watcher-operator and just adding. @tosky proposed https://github.com/openstack-k8s-operators/s2i-openstack-containers/pull/213/changes to stablish a naming convention which seems fine.

@SeanMooney SeanMooney Sep 18, 2026 •

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

sure <repo>-s2i-component-jobs is fine it helps us find them later.

@centosinfra-prod-github-app

Copy link
Copy Markdown

Zuul encountered a syntax error while parsing its
configuration in the repo openstack-k8s-operators/s2i-openstack-containers on branch main. The
problem was:

Job ^containers/watcher/s2i-openstack-container-consumer-smoke not defined

The problem appears in the "openstack-k8s-operators/s2i-openstack-containers" project stanza:

project:
name: openstack-k8s-operators/s2i-openstack-containers
github-check:
jobs:
- s2i-openstack-containers-molecule:
irrelevant-files: &irrelevant-for-zuul
- ^.github/.*
- ^docs/.*
- ^.*.md$
...

in "openstack-k8s-operators/s2i-openstack-containers/zuul.d/projects.yaml@main", line 2

@centosinfra-prod-github-app

Copy link
Copy Markdown

Zuul encountered a syntax error while parsing its
configuration in the repo openstack-k8s-operators/s2i-openstack-containers on branch main. The
problem was:

Job ^containers/watcher/s2i-openstack-container-consumer-smoke not defined

The problem appears in the "openstack-k8s-operators/s2i-openstack-containers" project stanza:

project:
name: openstack-k8s-operators/s2i-openstack-containers
github-check:
jobs:
- s2i-openstack-containers-molecule:
irrelevant-files: &irrelevant-for-zuul
- ^.github/.*
- ^docs/.*
- ^.*.md$
...

in "openstack-k8s-operators/s2i-openstack-containers/zuul.d/projects.yaml@main", line 2

@centosinfra-prod-github-app

Copy link
Copy Markdown

Zuul encountered a syntax error while parsing its
configuration in the repo openstack-k8s-operators/s2i-openstack-containers on branch main. The
problem was:

Job 2i-openstack-container-consumer-smoke not defined

The problem appears in the "openstack-k8s-operators/s2i-openstack-containers" project stanza:

project:
name: openstack-k8s-operators/s2i-openstack-containers
github-check:
jobs:
- s2i-openstack-containers-molecule:
irrelevant-files: &irrelevant-for-zuul
- ^.github/.*
- ^docs/.*
- ^.*.md$
...

in "openstack-k8s-operators/s2i-openstack-containers/zuul.d/projects.yaml@main", line 2

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/259d0d328fe64c7a8e328603debeb142

✔️ s2i-openstack-containers-molecule SUCCESS in 2m 06s
✔️ s2i-openstack-container-content-provider SUCCESS in 2h 51m 41s
✔️ s2i-openstack-container-consumer-smoke SUCCESS in 5m 50s
✔️ s2i-openstack-deploy-validation SUCCESS in 1h 27m 00s
❌ watcher-s2i-tempest RETRY_LIMIT in 18m 25s

@rlandy

rlandy commented Sep 17, 2026

Copy link
Copy Markdown
Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/b0a85b948fe44760bea78d92db6ad483

✔️ s2i-openstack-containers-molecule SUCCESS in 2m 03s
✔️ s2i-openstack-container-content-provider SUCCESS in 2h 18m 08s
✔️ s2i-openstack-container-consumer-smoke SUCCESS in 5m 50s
✔️ s2i-openstack-deploy-validation SUCCESS in 1h 26m 43s
❌ watcher-s2i-tempest RETRY_LIMIT in 16m 43s

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/45a9b5d201474b81bb0516278bb190a5

❌ s2i-openstack-containers-molecule NODE_FAILURE Node(set) request 100-0000205893 failed in 0s
❌ s2i-openstack-container-content-provider NODE_FAILURE Node(set) request 100-0000205894 failed in 0s
⚠️ s2i-openstack-container-consumer-smoke SKIPPED Skipped due to failed job s2i-openstack-container-content-provider
⚠️ s2i-openstack-deploy-validation SKIPPED Skipped due to failed job s2i-openstack-container-content-provider
⚠️ watcher-s2i-tempest SKIPPED Skipped due to failed job s2i-openstack-container-content-provider

@raukadah

Copy link
Copy Markdown
Contributor

recheck

Comment thread zuul.d/projects.yaml Outdated
@centosinfra-prod-github-app

Copy link
Copy Markdown

Zuul encountered a syntax error while parsing its
configuration in the repo openstack-k8s-operators/s2i-openstack-containers on branch main. The
problem was:

while parsing a block mapping
in "openstack-k8s-operators/s2i-openstack-containers/zuul.d/projects.yaml@main", line 23, column 11
did not find expected key
in "openstack-k8s-operators/s2i-openstack-containers/zuul.d/projects.yaml@main", line 26, column 13

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/53762df930d7402196454d1db3eb5464

✔️ s2i-openstack-containers-molecule SUCCESS in 2m 06s
✔️ s2i-openstack-container-content-provider SUCCESS in 2h 24m 04s
✔️ s2i-openstack-container-consumer-smoke SUCCESS in 6m 07s
✔️ s2i-openstack-deploy-validation SUCCESS in 1h 25m 48s
❌ watcher-s2i-tempest POST_FAILURE in 26m 52s

@rlandy
rlandy force-pushed the watcher-test branch 2 times, most recently from ea46276 to 1ea14ea Compare September 18, 2026 09:13
@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/4dd484408f04464ab54053d1da41b254

✔️ s2i-openstack-containers-molecule SUCCESS in 2m 04s
✔️ s2i-openstack-container-content-provider SUCCESS in 2h 57m 58s
✔️ s2i-openstack-container-consumer-smoke SUCCESS in 6m 10s
✔️ s2i-openstack-deploy-validation SUCCESS in 1h 25m 12s
❌ watcher-s2i-tempest FAILURE in 1h 47m 24s

Comment thread zuul.d/projects.yaml Outdated
watcher and base container changes should be gated
by Watcher jobs. This PR adds a template to include
Watcher testing on those directories.

Depends-On: openstack-k8s-operators/watcher-operator#457
@centosinfra-prod-github-app

Copy link
Copy Markdown

Zuul encountered a syntax error while parsing its
configuration in the repo openstack-k8s-operators/s2i-openstack-containers on branch main. The
problem was:

The project template "watcher-operator-s2i-component-jobs" was not
found.

The problem appears in the "openstack-k8s-operators/s2i-openstack-containers" project stanza:

project:
name: openstack-k8s-operators/s2i-openstack-containers
templates:
- watcher-operator-s2i-component-jobs
github-check:
jobs:
- s2i-openstack-containers-molecule:
irrelevant-files: &irrelevant-for-zuul
- ^.github/.*
...

in "openstack-k8s-operators/s2i-openstack-containers/zuul.d/projects.yaml@main", line 2

@rlandy

rlandy commented Sep 22, 2026

Copy link
Copy Markdown
Author

This will fail until the Depends-On changes has merged - expected with a template addition

@tosky

tosky commented Sep 22, 2026

Copy link
Copy Markdown
Contributor

This will fail until the Depends-On changes has merged - expected with a template addition

It looks like Depends-On works, but you need to set it in the PR description.
See for example:
#213

(I guess we need to coordinate the merges)

@rlandy

rlandy commented Sep 22, 2026

Copy link
Copy Markdown
Author

recheck

@rlandy

rlandy commented Sep 22, 2026

Copy link
Copy Markdown
Author

Thank you @tosky . checking run with Depends-On.

@SeanMooney SeanMooney left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

this looks correct but we need to be carful of the merge order.

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/be2325a44f934f15a69045b42adba1ee

❌ watcher-s2i-tempest FAILURE in 1h 51m 01s
✔️ s2i-openstack-containers-molecule SUCCESS in 2m 14s
✔️ s2i-openstack-container-content-provider SUCCESS in 8h 45m 17s
✔️ s2i-openstack-container-consumer-smoke SUCCESS in 5m 51s
❌ s2i-openstack-deploy-validation POST_FAILURE in 24m 37s

@amoralej

Copy link
Copy Markdown
Contributor

check-rdo

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c9a17626565943e082701077cc9f8594

⚠️ watcher-s2i-tempest SKIPPED Skipped due to failed job s2i-openstack-container-content-provider
❌ s2i-openstack-containers-molecule NODE_FAILURE Node(set) request 100-0000211141 failed in 0s
❌ s2i-openstack-container-content-provider NODE_FAILURE Node(set) request 100-0000211142 failed in 0s
⚠️ s2i-openstack-container-consumer-smoke SKIPPED Skipped due to failed job s2i-openstack-container-content-provider
⚠️ s2i-openstack-deploy-validation SKIPPED Skipped due to failed job s2i-openstack-container-content-provider

@amoralej

Copy link
Copy Markdown
Contributor

check-rdo

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/eff5935047c04ec8a7350a61b7997e0f

⚠️ watcher-s2i-tempest SKIPPED Skipped due to failed job s2i-openstack-container-content-provider
✔️ s2i-openstack-containers-molecule SUCCESS in 2m 25s
❌ s2i-openstack-container-content-provider NODE_FAILURE Node(set) request 100-0000211151 failed in 0s
⚠️ s2i-openstack-container-consumer-smoke SKIPPED Skipped due to failed job s2i-openstack-container-content-provider
⚠️ s2i-openstack-deploy-validation SKIPPED Skipped due to failed job s2i-openstack-container-content-provider

@amoralej

Copy link
Copy Markdown
Contributor

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/b1f48971045c4cb5866e5414808f69b7

✔️ watcher-s2i-tempest SUCCESS in 1h 56m 04s
✔️ s2i-openstack-containers-molecule SUCCESS in 2m 10s
✔️ s2i-openstack-container-content-provider SUCCESS in 4h 07m 08s
✔️ s2i-openstack-container-consumer-smoke SUCCESS in 6m 31s
❌ s2i-openstack-deploy-validation POST_FAILURE in 22m 09s

@amoralej

Copy link
Copy Markdown
Contributor

check-rdo

@centosinfra-prod-github-app

Copy link
Copy Markdown

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants