Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Zuul encountered a syntax error while parsing its The project template "manila-operator-s2i-component-jobs" was not The problem appears in the "openstack-k8s-operators/s2i-openstack-containers" project stanza: project: in "openstack-k8s-operators/s2i-openstack-containers/zuul.d/projects.yaml@main", line 2 |
8969cfc to
ca80925
Compare
|
Build failed (check pipeline). Post
|
|
recheck |
|
Build failed (check pipeline). Post ❌ manila-s2i-test-cephfs NODE_FAILURE Node(set) request 099-0000207262 failed in 0s |
|
recheck |
|
Build failed (check pipeline). Post
|
|
recheck |
|
Build failed (check pipeline). Post
|
ca80925 to
844516c
Compare
|
Build failed (check pipeline). Post ❌ cinder-s2i-test-ceph FAILURE in 1h 20m 48s |
|
recheck |
|
Build failed (check pipeline). Post ❌ cinder-s2i-test-ceph FAILURE in 1h 13m 48s |
|
recheck |
|
Build failed (check pipeline). Post ❌ cinder-s2i-test-ceph FAILURE in 1h 40m 12s |
|
recheck |
|
Build failed (check pipeline). Post ❌ cinder-s2i-test-ceph FAILURE in 1h 39m 29s |
|
recheck |
|
Build failed (check pipeline). Post ❌ cinder-s2i-test-ceph FAILURE in 1h 42m 03s |
|
recheck |
|
Merge Failed. This change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset. |
|
recheck |
|
This change depends on a change that failed to merge. Change openstack-k8s-operators/test-operator#522 is needed. |
|
recheck |
|
This change depends on a change that failed to merge. Change openstack-k8s-operators/test-operator#522 is needed. |
|
recheck |
Depends-On: #215
Depends-On: openstack-k8s-operators/ci-framework#4197
Depends-On: #234
Depends-On: openstack-k8s-operators/test-operator#522